Skip to content

Update go-git/v5 to latest release - #4050

Open
dsimansk wants to merge 2 commits into
knative:mainfrom
dsimansk:pr/upgrade-git-libs
Open

Update go-git/v5 to latest release#4050
dsimansk wants to merge 2 commits into
knative:mainfrom
dsimansk:pr/upgrade-git-libs

Conversation

@dsimansk

@dsimansk dsimansk commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

Changes

  • Update go-git/v5 to latest release

/kind

Fixes #

Release Note

Update go-git/v5 to latest release

Maintenance and trying to verify it's working without additional changes for midstream purposes.
/cc @matejvasek @gauron99

@knative-prow

knative-prow Bot commented Sep 8, 2026

Copy link
Copy Markdown

@dsimansk: The label(s) kind/<kind> cannot be applied, because the repository doesn't have them.

Details

In response to this:

Changes

  • Update go-git/v5 to latest release

/kind

Fixes #

Release Note

Update go-git/v5 to latest release

Maintenance and trying to verify it's working without additional changes for midstream purposes.
/cc @matejvasek @gauron99

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@knative-prow

knative-prow Bot commented Sep 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added approved 🤖 PR has been approved by an approver from all required OWNERS files. size/M 🤖 PR changes 30-99 lines, ignoring generated files. labels Sep 8, 2026
return fmt.Errorf("failed to get worktree: %w", err)
}
fs = filesystem.NewBillyFilesystem(wt.Filesystem)
fs = filesystem.NewOsFilesystem(tempDir)

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@matejvasek PTAL, this is necessary change because of worktree hardening in go-git. It no longer allow .git to be .CopyFromFS later in the function. The change keeps the intention to copy the whole repo content with .git and its metadata preserved.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved 🤖 PR has been approved by an approver from all required OWNERS files. size/M 🤖 PR changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant