Skip to content

feat: update sphinxbase to 0.8+5prealpha+1-16 - #2

Merged
Zeno-sole merged 11 commits into
masterfrom
dcbot/debian/0.8+5prealpha+1
Sep 23, 2026
Merged

Zeno-sole merged 11 commits into
masterfrom
dcbot/debian/0.8+5prealpha+1

Conversation

@deepin-community-bot

Copy link
Copy Markdown
Contributor

This pull request is requested by @UTsweetyfish.

Basic Information

Old Version: 0.8+5prealpha+1-13deepin1
New Version: 0.8+5prealpha+1-16
Old version may contain dde / deepin patches. Please review more precisely.

Patch series

--- a/debian/patches/series
+++ b/debian/patches/series
@@ -5,3 +5,4 @@
 log_shifted_ignore
 debug
 python3.10
+utf-8

@github-actions

github-actions Bot commented Sep 22, 2026

Copy link
Copy Markdown

TAG Bot

TAG: 0.8+5prealpha+1-16deepin1
EXISTED: no
DISTRIBUTION: unstable

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign qaqland for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@Zeno-sole

Copy link
Copy Markdown

/integrate

@github-actions

Copy link
Copy Markdown

AutoIntegrationPr Bot
auto integrate with pr url: deepin-community/Repository-Integration#4589
PrNumber: 4589
PrBranch: auto-integration-35729481368

@UTsweetyfish
UTsweetyfish marked this pull request as ready for review September 22, 2026 13:29
@UTsweetyfish

Copy link
Copy Markdown
Contributor

/integrate

@Zeno-sole
Zeno-sole merged commit 5b7bf38 into master Sep 23, 2026
6 of 10 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants