Skip to content

feat: update pptp-linux to 1.10.0-2 - #1

Merged
Zeno-sole merged 1 commit into
masterfrom
dcbot/debian/1.10.0
Sep 22, 2026
Merged

Zeno-sole merged 1 commit into
masterfrom
dcbot/debian/1.10.0

Conversation

@deepin-community-bot

Copy link
Copy Markdown
Contributor

This pull request is requested by @UTsweetyfish.

Basic Information

Old Version: 1.10.0-1
New Version: 1.10.0-2

Patch series

--- a/debian/patches/series
+++ b/debian/patches/series
@@ -1 +1,14 @@
-debian-569964-pptp-manpage.diff
+
+# Reverted upstream commits
+
+# Cherry-picked commits. Keep in upstream's chronological order
+cherry-pick/1558124752.1.10.0-2-g5269b57.fix-parallel-build-issue.patch
+cherry-pick/1575418932.1.10.0-4-gf449742.pptsetup-do-not-mess-up-the-password-when-reading-from-stdin.patch
+cherry-pick/1589514483.1.10.0-6-gff82566.add-missing-manual-page-options-and-fix-typo.patch
+cherry-pick/1599550704.1.10.0-7-g194620e.feature-terminate-from.patch
+
+# Patches that should go upstream
+
+# Debian-specific modifications
+debian/manpage-doc-location.patch
+debian/allow-rootless-build.patch

@github-actions

Copy link
Copy Markdown

TAG Bot

TAG: 1.10.0-2
EXISTED: no
DISTRIBUTION: unstable

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign zeno-sole for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@UTsweetyfish
UTsweetyfish marked this pull request as ready for review September 22, 2026 10:08
@UTsweetyfish

Copy link
Copy Markdown

/integrate

@github-actions

Copy link
Copy Markdown

AutoIntegrationPr Bot
auto integrate with pr url: deepin-community/Repository-Integration#4586
PrNumber: 4586
PrBranch: auto-integration-35714366430

@Zeno-sole
Zeno-sole merged commit 007c845 into master Sep 22, 2026
6 of 10 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants