feat: update libpcap to 1.10.5-2 - #1
deepin-community-bot[bot] wants to merge 2 commits into
Conversation
|
TAG Bot TAG: 1.10.5-2deepin1 |
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
TAG Bot New tag: 1.10.1-4deepin1 |
|
TAG Bot New tag: 1.10.1-4deepin3 |
|
TAG Bot New tag: 1.10.1-4deepin4 |
This pull request is requested by @zengwei00.
Basic Information
Old Version: 1.10.1-4
New Version: 1.10.5-2
Potential transition
64-bit time_t transition
Caution
DCBot assumes this package is already gone through 64-bit time_t transition.
Patch series