Skip to content

feat: update bind9 to 1:9.20.29-1~deb13u1 - #16

Merged
Zeno-sole merged 2 commits into
masterfrom
dcbot/debian/9.20.29
Sep 23, 2026
Merged

Zeno-sole merged 2 commits into
masterfrom
dcbot/debian/9.20.29

Conversation

@deepin-community-bot

Copy link
Copy Markdown
Contributor

This pull request is requested by @UTsweetyfish.

Basic Information

Old Version: 1:9.20.23-1~deb13u1deepin12
New Version: 1:9.20.29-1~deb13u1
Old version may contain dde / deepin patches. Please review more precisely.

Patch series

--- a/debian/patches/series
+++ b/debian/patches/series
@@ -1,20 +1,2 @@
 0001-Disable-treat-warnings-as-errors-in-sphinx-build.patch
 0002-Disable-RTLD_DEEPBIND-in-Samba-DLZ-module.patch
-CVE-2026-13204.patch
-CVE-2026-12617.patch
-CVE-2026-11622.patch
-CVE-2026-13321.patch
-CVE-2026-11605.patch
-CVE-2026-11331.patch
-CVE-2026-10822.patch
-CVE-2026-10723.patch
-CVE-2026-11721.patch
-CVE-2026-76163.patch
-CVE-2026-77692.patch
-CVE-2026-19666.patch
-CVE-2026-81563.patch
-CVE-2026-19941.patch
-CVE-2026-77119.patch
-CVE-2026-75029.patch
-CVE-2026-19668.patch
-CVE-2026-19033.patch

@github-actions

github-actions Bot commented Sep 22, 2026

Copy link
Copy Markdown

TAG Bot

TAG: 1%9.20.29-1_deb13u1deepin1
EXISTED: no
DISTRIBUTION: unstable

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign yukarichiba for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@UTsweetyfish

Copy link
Copy Markdown
Contributor

/integrate

@github-actions

Copy link
Copy Markdown

AutoIntegrationPr Bot
auto integrate with pr url: deepin-community/Repository-Integration#4593
PrNumber: 4593
PrBranch: auto-integration-35750335132

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

/hold
因为该quilt包的上游版本号变更,详情见: deepin-community/infra-settings#134

@Zeno-sole
Zeno-sole merged commit 7c11f83 into master Sep 23, 2026
6 of 10 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants