feat: update bind9 to 1:9.20.26-1~deb13u1 - #11
deepin-community-bot[bot] wants to merge 1 commit into
Conversation
|
TAG Bot TAG: 1%9.20.26-1_deb13u1 |
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
| bind9 (1:9.20.23-1~deb13u1deepin1) unstable; urgency=medium | ||
|
|
||
| [ lichenggang ] | ||
| * Disable Build-Dep xindy for sunway. | ||
|
|
||
| -- Tianyu Chen <sweetyfish@deepin.org> Thu, 21 May 2026 15:38:10 +0800 |
| @@ -37,7 +37,7 @@ Build-Depends-Indep: fonts-freefont-otf, | |||
| texlive-fonts-extra, | |||
| texlive-latex-recommended, | |||
| texlive-xetex, | |||
| xindy [!sw64] | |||
| xindy | |||
|
TAG Bot New tag: 1%9.20.23-1_deb13u1deepin12 |
This pull request is requested by @UTsweetyfish.
Basic Information
Old Version:
1:9.20.23-1~deb13u1deepin10New Version:
1:9.20.26-1~deb13u1Old version may contain dde / deepin patches. Please review more precisely.
Patch series