Skip to content

auto integration deepin-community/openvpn - #4570

Open
deepin-community-ci-bot[bot] wants to merge 1 commit into
masterfrom
auto-integration-35481409057
Open

deepin-community-ci-bot[bot] wants to merge 1 commit into
masterfrom
auto-integration-35481409057

Conversation

@deepin-community-ci-bot

Copy link
Copy Markdown
Contributor

auto integration from pr deepin-community/openvpn#16.

Signed-off-by: github-actions <github-actions@users.noreply.github.com>
@github-actions

Copy link
Copy Markdown

RepoTagCheck Bot

Listed repo integration are not modified changelog:

repo tagOrsha
deepin-community/openvpn 3fe7327fa3ca8522547bed74d3c5b7795498a4ef

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign tsic404 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@deepin-ci-robot

Copy link
Copy Markdown
Contributor

Hi @deepin-community-ci-bot[bot]. Thanks for your PR.

I'm waiting for a deepin-community member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@github-actions

Copy link
Copy Markdown

INTEGRATION Bot

Integration Details: fix(cve): CVE-2026-84471 - Move check_session_buf_not_used method into the method that free the buf by deepin-ci-robot@deepin-community/openvpn by deepin-community-ci-bot[bot]
@deepin-community/ci CVE detected

@github-actions

Copy link
Copy Markdown

IntegrationProjector Bot
Deepin Testing Integration Project Manager Info
IssueID: 14008
ProjectItemID: PVTI_lADOABhNec4AL0Jezg7x7TA
IssueAddr: linuxdeepin/developer-center#14008

This branch is waiting to be deployed

1 waiting deployment
TestingIntegrationTesterReview 6ed1a2da Waiting Sep 20, 2026 by deepin-community-ci-bot[bot] via pass integration / auto test #7203
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants