Skip to content

Create CODEOWNERS - #210

Open
joaorosado wants to merge 2 commits into
masterfrom
chore/add-codeowners
Open

joaorosado wants to merge 2 commits into
masterfrom
chore/add-codeowners

Conversation

@joaorosado

Copy link
Copy Markdown

Description

Adding O11 Frontend Runtime as code owners of the current default branch as well

@rmb-guerra

Copy link
Copy Markdown

this is not the 'main' branch. this PR must be done against branch 'b11'

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants