Skip to content

chore: bump debian-base and git-sync - #2309

Merged
google-oss-prow[bot] merged 1 commit into
GoogleContainerTools:mainfrom
Camila-B:fix-vulns
Oct 2, 2026
Merged

google-oss-prow[bot] merged 1 commit into
GoogleContainerTools:mainfrom
Camila-B:fix-vulns

Conversation

@Camila-B

@Camila-B Camila-B commented Oct 2, 2026

Copy link
Copy Markdown
Contributor
  • debian-base: trixie-v1.0.0-gke.3 -> trixie-v1.0.0-gke.4
  • git-sync: v4.5.0-gke.2 -> v4.5.0-gke.3

- debian-base: trixie-v1.0.0-gke.3 -> trixie-v1.0.0-gke.4
- git-sync: v4.5.0-gke.2 -> v4.5.0-gke.3

@tiffanny29631 tiffanny29631 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: tiffanny29631

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-prow
google-oss-prow Bot merged commit 17dd6d7 into GoogleContainerTools:main Oct 2, 2026
12 checks passed
tiffanny29631 pushed a commit to tiffanny29631/kpt-config-sync that referenced this pull request Oct 6, 2026
- debian-base: trixie-v1.0.0-gke.3 -> trixie-v1.0.0-gke.4
- git-sync: v4.5.0-gke.2 -> v4.5.0-gke.3
google-oss-prow Bot pushed a commit that referenced this pull request Oct 7, 2026
* chore: bump helm, base image and git-sync (#2308)

* chore: bump debian-base and git-sync (#2309)

- debian-base: trixie-v1.0.0-gke.3 -> trixie-v1.0.0-gke.4
- git-sync: v4.5.0-gke.2 -> v4.5.0-gke.3

* chore: bump otelcontribcol (#2311)

---------

Co-authored-by: Camila Bustos <camilabustos@google.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants