Repository navigation
chore: bump cloud.google.com/go/auth from 0.23.3 to 0.24.0 - #2305
dependabot[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a GoogleContainerTools member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/ok-to-test |
Bumps [cloud.google.com/go/auth](https://github.com/googleapis/google-cloud-go) from 0.23.3 to 0.24.0. - [Release notes](https://github.com/googleapis/google-cloud-go/releases) - [Changelog](https://github.com/googleapis/google-cloud-go/blob/main/CHANGES.md) - [Commits](googleapis/google-cloud-go@auth/v0.23.3...v0.24.0) --- updated-dependencies: - dependency-name: cloud.google.com/go/auth dependency-version: 0.24.0 dependency-type: direct:production update-type: version-update:semver-minor ... Signed-off-by: dependabot[bot] <support@github.com>
b14bd8a to
bdbdfc0
Compare
Bumps cloud.google.com/go/auth from 0.23.3 to 0.24.0.
Changelog
Sourced from cloud.google.com/go/auth's changelog.
... (truncated)
Commits
777200cREADME.md: v0.24.0 release notes88ee2dfRemoved Blank line47b2ba0added app profile crudl for cbtc4af15fRefactoring of parseColumnsFilter function.051b66cBigQuery: Add ORC as DataFormat for load operations.ae1e370profiler: use readable runID in e2e testsf5adaf4profiler: busybench should exit after finishing profilingb75b1b1internal: add doc.go to fix failing build8e74ea3internal: test against 1.10 on travis87a3e13debug: apply linux build constraint to dependencies