Skip to content

Make sure the complyctl directory exists - #15141

Open
jan-cerny wants to merge 1 commit into
ComplianceAsCode:masterfrom
jan-cerny:ansible_scan_baseline
Open

jan-cerny wants to merge 1 commit into
ComplianceAsCode:masterfrom
jan-cerny:ansible_scan_baseline

Conversation

@jan-cerny

Copy link
Copy Markdown
Member

Creates the /root/.complytime directory in the vagrant box if it doesn't exist.

Addressing:

TASK [[ LOW ] Write complytime.yaml on the VM] **********************************************************************************************
[ERROR]: Task failed: Module failed: Destination directory /root/.complytime does not exist
Origin: /home/jcerny/work/git/scap-security-guide/utils/nist_sync/ansible/tasks/scan_baseline.yml:53:3

51 # 2. Configure complyctl on the VM
52 # ---------------------------------------------------------------------------
53 - name: "[ {{ baseline_name | upper }} ] Write complytime.yaml on the VM"
     ^ column 3

fatal: [nist-rhel9-scanner]: FAILED! => {"changed": false, "checksum": "7d7f56be0e6ea630aa7aed03caee0c18a6a3129c", "msg": "Destination directory /root/.complytime does not exist"}

Review Hints:

Follow the instructions in section "Running everything with Vagrant" in docs/manual/developer/15_nist_800_53_gemara_walkthrough.md

Creates the `/root/.complytime` directory in the vagrant box
if it doesn't exist.

Addressing:

```
TASK [[ LOW ] Write complytime.yaml on the VM] **********************************************************************************************
[ERROR]: Task failed: Module failed: Destination directory /root/.complytime does not exist
Origin: /home/jcerny/work/git/scap-security-guide/utils/nist_sync/ansible/tasks/scan_baseline.yml:53:3

51 # 2. Configure complyctl on the VM
52 # ---------------------------------------------------------------------------
53 - name: "[ {{ baseline_name | upper }} ] Write complytime.yaml on the VM"
     ^ column 3

fatal: [nist-rhel9-scanner]: FAILED! => {"changed": false, "checksum": "7d7f56be0e6ea630aa7aed03caee0c18a6a3129c", "msg": "Destination directory /root/.complytime does not exist"}
```
@jan-cerny jan-cerny added this to the 0.1.83 milestone Sep 24, 2026
@openshift-ci

openshift-ci Bot commented Sep 24, 2026

Copy link
Copy Markdown

@jan-cerny: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/4.12-images 2891aac link true /test 4.12-images

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant